Skip to content

[main] Update community files - #731

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-community-main
Oct 9, 2026
Merged

knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-community-main

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

new repo -aliok

Produced by: knative-extensions/knobots/actions/update-community

Details:


Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow knative-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Oct 9, 2026
@knative-prow

knative-prow Bot commented Oct 9, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@knative-prow

knative-prow Bot commented Oct 9, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign matzew for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow
knative-prow Bot requested review from lberk and matzew October 9, 2026 09:25
@knative-prow
knative-prow Bot merged commit 6fe3749 into knative-extensions:main Oct 9, 2026
14 checks passed
@codecov

codecov Bot commented Oct 9, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 20.09%. Comparing base (5a7e055) to head (63d565e).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##             main     #731   +/-   ##
=======================================
  Coverage   20.09%   20.09%           
=======================================
  Files          34       34           
  Lines        1204     1204           
=======================================
  Hits          242      242           
  Misses        958      958           
  Partials        4        4           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant