Skip to content

Fix non-contant format string err in go 1.26 - #694

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
dsimansk:pr/fix-format-string
Sep 18, 2026
Merged

knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
dsimansk:pr/fix-format-string

Conversation

@dsimansk

Copy link
Copy Markdown
Contributor

Fixes #

Proposed Changes

  • Fix non-contant format string err in go 1.26

Release Note

Fix non-contant format string err in go 1.26

Related to knative/pkg#3392

/cc @gauron99

@knative-prow
knative-prow Bot requested a review from gauron99 September 18, 2026 15:59
@knative-prow knative-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 18, 2026
@knative-prow

knative-prow Bot commented Sep 18, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 18, 2026
@codecov

codecov Bot commented Sep 18, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 37.93%. Comparing base (006534b) to head (c07e517).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##             main     #694   +/-   ##
=======================================
  Coverage   37.93%   37.93%           
=======================================
  Files          20       20           
  Lines         580      580           
=======================================
  Hits          220      220           
  Misses        347      347           
  Partials       13       13           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@gauron99

Copy link
Copy Markdown

/lgtm

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 18, 2026
@knative-prow
knative-prow Bot merged commit aa20735 into knative-extensions:main Sep 18, 2026
20 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants