Skip to content

Fix non-contant format string err in go 1.26 - #1152

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
dsimansk:pr/fix-format-string
Sep 19, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
dsimansk:pr/fix-format-string

Conversation

@dsimansk

Copy link
Copy Markdown
Contributor

Changes

  • Fix non-contant format string err in go 1.26

/kind

Fixes #

Release Note

Fix non-contant format string err in go 1.26

/cc @gauron99 @dprotaso

Related to knative/pkg#3392

@knative-prow

knative-prow Bot commented Sep 18, 2026

Copy link
Copy Markdown

@dsimansk: The label(s) kind/<kind> cannot be applied, because the repository doesn't have them.

Details

In response to this:

Changes

  • Fix non-contant format string err in go 1.26

/kind

Fixes #

Release Note

Fix non-contant format string err in go 1.26

/cc @gauron99 @dprotaso

Related to knative/pkg#3392

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@knative-prow
knative-prow Bot requested review from dprotaso and gauron99 September 18, 2026 18:37
@knative-prow knative-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 18, 2026
@knative-prow

knative-prow Bot commented Sep 18, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 18, 2026
@dsimansk dsimansk added kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. and removed approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Sep 18, 2026
@codecov

codecov Bot commented Sep 18, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 93.34%. Comparing base (17b28c6) to head (f77d68b).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##             main    #1152   +/-   ##
=======================================
  Coverage   93.34%   93.34%           
=======================================
  Files          35       35           
  Lines        1007     1007           
=======================================
  Hits          940      940           
  Misses         53       53           
  Partials       14       14           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@gauron99

Copy link
Copy Markdown

/lgtm

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 19, 2026
@dsimansk

Copy link
Copy Markdown
Contributor Author

/approved

@dsimansk dsimansk added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 19, 2026
@knative-prow
knative-prow Bot merged commit 7d916c7 into knative:main Sep 19, 2026
27 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants