Skip to content

[release-1.19] Upgrade to latest dependencies [skip-dot-release] - #3243

Merged
knative-prow[bot] merged 1 commit into
knative:release-1.19from
knative-automation:auto-updates/update-deps-release-1.19
Sep 2, 2025
Merged

knative-prow[bot] merged 1 commit into
knative:release-1.19from
knative-automation:auto-updates/update-deps-release-1.19

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

fix dot release script -dprotaso

/cc knative/serving-writers knative/eventing-writers
/assign knative/serving-writers knative/eventing-writers

Produced by: knative-extensions/knobots/actions/update-deps

bumping knative.dev/hack f88b7db...af735b2:
  > af735b2 Fix dot releases (# 434)

Signed-off-by: Knative Automation <automation@knative.team>
@knative-automation
knative-automation force-pushed the auto-updates/update-deps-release-1.19 branch from cd87bc0 to d14860f Compare September 2, 2025 15:49
@knative-prow
knative-prow Bot requested review from a team September 2, 2025 15:49
@knative-prow knative-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 2, 2025
@knative-prow
knative-prow Bot merged commit 9aa1144 into knative:release-1.19 Sep 2, 2025
25 checks passed
@knative-prow

knative-prow Bot commented Sep 2, 2025

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign creydr for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@dprotaso dprotaso changed the title [release-1.19] Upgrade to latest dependencies [release-1.19] Upgrade to latest dependencies [skip-dot-release] Sep 2, 2025
@codecov

codecov Bot commented Sep 2, 2025

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 74.61%. Comparing base (86f88f6) to head (d14860f).
⚠️ Report is 2 commits behind head on release-1.19.

Additional details and impacted files
@@               Coverage Diff                @@
##           release-1.19    #3243      +/-   ##
================================================
- Coverage         74.64%   74.61%   -0.03%     
================================================
  Files               209      209              
  Lines             12025    12025              
================================================
- Hits               8976     8973       -3     
- Misses             2770     2772       +2     
- Partials            279      280       +1     

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/S Denotes a PR that changes 10-29 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant