Repository navigation
Bump the otel group across 1 directory with 13 updates - #3388
dependabot[bot] wants to merge 2 commits into
Conversation
|
There are empty aliases in OWNER_ALIASES, cleanup is advised. |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #3388 +/- ##
==========================================
+ Coverage 74.83% 74.85% +0.02%
==========================================
Files 189 189
Lines 8349 8349
==========================================
+ Hits 6248 6250 +2
+ Misses 1855 1854 -1
+ Partials 246 245 -1 ☔ View full report in Codecov by Harness. 🚀 New features to boost your workflow:
|
|
@dependabot recreate |
|
rebase DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
f985ccc to
e437e18
Compare
|
There are empty aliases in OWNER_ALIASES, cleanup is advised. |
|
@dependabot recreate |
4a7a4f4 to
4103fa5
Compare
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
There are empty aliases in OWNER_ALIASES, cleanup is advised. |
|
@dependabot recreate |
|
rebase DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Bumps the otel group with 13 updates in the / directory: | Package | From | To | | --- | --- | --- | | [go.opentelemetry.io/contrib/instrumentation/net/http/otelhttp](https://github.com/open-telemetry/opentelemetry-go-contrib) | `0.70.0` | `0.72.0` | | [go.opentelemetry.io/contrib/instrumentation/runtime](https://github.com/open-telemetry/opentelemetry-go-contrib) | `0.70.0` | `0.72.0` | | [go.opentelemetry.io/otel](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetricgrpc](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetrichttp](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracegrpc](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/exporters/prometheus](https://github.com/open-telemetry/opentelemetry-go) | `0.67.0` | `0.69.0` | | [go.opentelemetry.io/otel/exporters/stdout/stdouttrace](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/metric](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/sdk](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/sdk/metric](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | | [go.opentelemetry.io/otel/trace](https://github.com/open-telemetry/opentelemetry-go) | `1.45.0` | `1.47.0` | Updates `go.opentelemetry.io/contrib/instrumentation/net/http/otelhttp` from 0.70.0 to 0.72.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go-contrib/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go-contrib/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go-contrib@zpages/v0.70.0...zpages/v0.72.0) Updates `go.opentelemetry.io/contrib/instrumentation/runtime` from 0.70.0 to 0.72.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go-contrib/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go-contrib/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go-contrib@zpages/v0.70.0...zpages/v0.72.0) Updates `go.opentelemetry.io/otel` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetricgrpc` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetrichttp` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracegrpc` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/exporters/prometheus` from 0.67.0 to 0.69.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@metric/x/v0.67.0...metric/x/v0.69.0) Updates `go.opentelemetry.io/otel/exporters/stdout/stdouttrace` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/metric` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/sdk` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/sdk/metric` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) Updates `go.opentelemetry.io/otel/trace` from 1.45.0 to 1.47.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.45.0...v1.47.0) --- updated-dependencies: - dependency-name: go.opentelemetry.io/contrib/instrumentation/net/http/otelhttp dependency-version: 0.71.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/contrib/instrumentation/runtime dependency-version: 0.71.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetricgrpc dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetrichttp dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracegrpc dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/exporters/prometheus dependency-version: 0.68.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/exporters/stdout/stdouttrace dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/metric dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/sdk dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/sdk/metric dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel - dependency-name: go.opentelemetry.io/otel/trace dependency-version: 1.46.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: otel ... Signed-off-by: dependabot[bot] <support@github.com>
8b1dd15 to
f709e3a
Compare
|
There are empty aliases in OWNER_ALIASES, cleanup is advised. |
Bumps the otel group with 13 updates in the / directory:
0.70.00.72.00.70.00.72.01.45.01.47.01.45.01.47.01.45.01.47.01.45.01.47.01.45.01.47.00.67.00.69.01.45.01.47.01.45.01.47.01.45.01.47.01.45.01.47.01.45.01.47.0Updates
go.opentelemetry.io/contrib/instrumentation/net/http/otelhttpfrom 0.70.0 to 0.72.0Changelog
Sourced from go.opentelemetry.io/contrib/instrumentation/net/http/otelhttp's changelog.
... (truncated)
Commits
0c76f61Release v1.47.0/v2.6.0/v0.72.0/v0.38.0/v0.27.0/v0.21.0/v0.17.0/v0.19.0 (#9802)318f5f9Add timeout on Kubernetes API calls for EKS detector (#9419)f848a15test(detector/ecs): increase unit test coverage (#9774)c0d84deRestore gofumpt clothe-returns and balance-calls (#9793)6215054fix(deps): update aws-sdk-go-v2 monorepo (#9785)9a2e8eafix(deps): update kubernetes monorepo to v0.37.13d8e338chore(deps): update sigs.k8s.io/json digest to 11ed52ebe5b688chore(deps): update go-openapi packagese382b4dchore(deps): update module github.com/fxamacker/cbor/v2 to v2.9.497f9e88fix(deps): update module github.com/moby/moby/client to v0.6.1Updates
go.opentelemetry.io/contrib/instrumentation/runtimefrom 0.70.0 to 0.72.0Changelog
Sourced from go.opentelemetry.io/contrib/instrumentation/runtime's changelog.
... (truncated)
Commits
0c76f61Release v1.47.0/v2.6.0/v0.72.0/v0.38.0/v0.27.0/v0.21.0/v0.17.0/v0.19.0 (#9802)318f5f9Add timeout on Kubernetes API calls for EKS detector (#9419)f848a15test(detector/ecs): increase unit test coverage (#9774)c0d84deRestore gofumpt clothe-returns and balance-calls (#9793)6215054fix(deps): update aws-sdk-go-v2 monorepo (#9785)9a2e8eafix(deps): update kubernetes monorepo to v0.37.13d8e338chore(deps): update sigs.k8s.io/json digest to 11ed52ebe5b688chore(deps): update go-openapi packagese382b4dchore(deps): update module github.com/fxamacker/cbor/v2 to v2.9.497f9e88fix(deps): update module github.com/moby/moby/client to v0.6.1Updates
go.opentelemetry.io/otelfrom 1.45.0 to 1.47.0Release notes
Sourced from go.opentelemetry.io/otel's releases.
... (truncated)
Changelog
Sourced from go.opentelemetry.io/otel's changelog.
... (truncated)
Commits
66cfc95Release 1.47.0/0.69.0/0.23.0/0.1.0 (#9024)f09d1deMerge commit from fork017b08cMerge commit from forkd9d3ac2Update ubuntu:26.04 Docker digest to 88a381db240391Update module github.com/uudashr/iface to v1.5.355899e3Fix zero attribute value length for malformed UTF-8 (#9054)75b1fe3Update github.com/charmbracelet/ultraviolet digest to 8786532cce78adUpdate github.com/golangci/rowserrcheck digest to f772fe6 (#9066)cdc6ee4Update module github.com/pjbgf/sha1cd to v0.7.0ef11bfcchore: Replace deprecated gofumpt extra-rules with extra options (#9062)Updates
go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetricgrpcfrom 1.45.0 to 1.47.0Release notes
Sourced from go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetricgrpc's releases.
... (truncated)
Changelog
Sourced from go.opentelemetry.io/otel/exporters/otlp/otlpmetric/otlpmetricgrpc's changelog.