Skip to content

Update owners files to use SC alias instead of TOC - #3396

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
dsimansk:pr/update-owners
Oct 9, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
dsimansk:pr/update-owners

Conversation

@dsimansk

@dsimansk dsimansk commented Oct 9, 2026

Copy link
Copy Markdown
Contributor

Changes

  • Update owners files to use SC alias instead of TOC
NONE

/cc @knative/steering-committee

@knative-prow
knative-prow Bot requested a review from a team October 9, 2026 09:52
@knative-prow

knative-prow Bot commented Oct 9, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@knative-prow knative-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Oct 9, 2026
Comment thread apis/duck/OWNERS
# The OWNERS file is used by prow to automatically merge approved PRs.

approvers:
- steering-committee

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

In theory, SC should be inherited from apis/OWNERS, but rather added explicitly.

@aliok

aliok commented Oct 9, 2026

Copy link
Copy Markdown
Member

/lgtm
/approve

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Oct 9, 2026
@codecov

codecov Bot commented Oct 9, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 74.82%. Comparing base (8c9ee30) to head (0d32fe2).
⚠️ Report is 2 commits behind head on main.

Additional details and impacted files
@@            Coverage Diff             @@
##             main    #3396      +/-   ##
==========================================
- Coverage   74.88%   74.82%   -0.06%     
==========================================
  Files         189      189              
  Lines        8349     8349              
==========================================
- Hits         6252     6247       -5     
- Misses       1853     1856       +3     
- Partials      244      246       +2     

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@creydr creydr left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@knative-prow

knative-prow Bot commented Oct 9, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: aliok, creydr, dsimansk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 9, 2026
@creydr

creydr commented Oct 9, 2026

Copy link
Copy Markdown
Member

/retest

@knative-prow
knative-prow Bot merged commit d742f3f into knative:main Oct 9, 2026
34 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants